Skip to content

kubernetes/examples: create e2e test job - #37422

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
justinsb:examples_test_with_kind
Jul 10, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
justinsb:examples_test_with_kind

Conversation

@justinsb

@justinsb justinsb commented Jul 9, 2026

Copy link
Copy Markdown
Member

We create a generic tests/e2e/e2e-with-kind entrypoint to avoid
having to create a new job for each example.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/config Issues or PRs related to code in /config area/jobs area/testgrid sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Jul 9, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from mattfarina and wojtek-t July 9, 2026 13:56
@justinsb

justinsb commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/assign @soltysh

@justinsb

justinsb commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

(As requested in kubernetes/examples#595 (comment) )

- wrapper.sh
- bash
- -c
- tests/e2e/e2e-with-kind

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

In terms of entrypoint naming, this has to match the PR kubernetes/examples#595, but we do otherwise have a free choice here.

One thing I've been doing in other projects is to put all the presubmit tests in dev/ci/presubmits/* , and then e.g. dev/ci/presubmits/e2e-with-kind would map to a test pull-examples-e2e-with-kind (for example)

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

One minor nit, but overall

/lgtm
/approve

Comment thread config/jobs/kubernetes/examples/examples-presubmits.yaml Outdated
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 10, 2026
@soltysh

soltysh commented Jul 10, 2026

Copy link
Copy Markdown
Contributor

/assign @dims

@kubernetes-prow kubernetes-prow Bot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 10, 2026
We create a generic tests/e2e/e2e-with-kind entrypoint to avoid
having to create a new job for each example.

Co-authored-by: Maciej Szulik <soltysh@gmail.com>
@justinsb
justinsb force-pushed the examples_test_with_kind branch from 616971d to 652640c Compare July 10, 2026 14:47
@justinsb

Copy link
Copy Markdown
Member Author

Applied suggestion and updated the other "e2e-with-kind" -> "e2e-kind". Thanks @soltysh :-)

@soltysh

soltysh commented Jul 10, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 10, 2026
@aojea

aojea commented Jul 10, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aojea, justinsb, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 10, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 0e9cc9b into kubernetes:master Jul 10, 2026
6 checks passed
@kubernetes-prow

Copy link
Copy Markdown
Contributor

@justinsb: Updated the job-config configmap in namespace default at cluster test-infra-trusted using the following files:

  • key examples-presubmits.yaml using file config/jobs/kubernetes/examples/examples-presubmits.yaml
Details

In response to this:

We create a generic tests/e2e/e2e-with-kind entrypoint to avoid
having to create a new job for each example.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs area/testgrid cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants