chore(deps): update nx to v23 (major) - #33
Open
github-actions[bot] wants to merge 1 commit into
Open
github-actions[bot] wants to merge 1 commit into
github-actions[bot] wants to merge 1 commit into
Conversation
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
December 23, 2025 06:36
ecf1cc6 to
8417a07
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
January 15, 2026 06:39
27f4664 to
7621b0e
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
3 times, most recently
from
January 30, 2026 06:51
3f40aa8 to
647da9d
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
3 times, most recently
from
February 7, 2026 06:48
6acf189 to
97280cc
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
February 17, 2026 07:05
b04bb13 to
32765c7
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
March 2, 2026 07:05
10549d4 to
5662f16
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
March 13, 2026 08:54
b037893 to
c272604
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
March 24, 2026 07:11
223ba47 to
bcea52c
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
April 5, 2026 07:11
e5b0953 to
2324026
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
April 14, 2026 07:37
2324026 to
f85cdcd
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
May 2, 2026 07:50
c076652 to
06babdc
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
May 18, 2026 10:07
06babdc to
7370d06
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
3 times, most recently
from
May 31, 2026 08:51
64d2a8e to
c161e32
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
June 4, 2026 10:04
c161e32 to
5380a93
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
June 20, 2026 09:02
5380a93 to
9172bb4
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
June 27, 2026 08:42
9172bb4 to
f9a8236
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
July 14, 2026 08:08
f9a8236 to
6cf2f9d
Compare
github-actions
Bot
force-pushed
the
renovate/major-nx
branch
from
July 17, 2026 08:16
6cf2f9d to
14c048b
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
^21.3.5→^23.0.0^21.3.5→^23.0.0^22.1.3→^23.0.0^21.3.5→^23.0.0^21.3.5→^23.0.0^21.3.5→^23.0.0^21.3.5→^23.0.0^21.3.5→^23.0.0Release Notes
nrwl/nx (@nx/angular)
v23.2.1Compare Source
23.2.1 (2026-09-09)
🚀 Features
🩹 Fixes
❤️ Thank You
v23.2.0Compare Source
23.2.0 (2026-09-02)
🚀 Features
🩹 Fixes
ng-packagrlazily (#36632)nx affected(#36569, #36568)@find(#36526)❤️ Thank You
v23.1.3Compare Source
23.1.3 (2026-08-31)
🩹 Fixes
ng-packagrlazily (#36632)❤️ Thank You
v23.1.2Compare Source
23.1.2 (2026-08-26)
🩹 Fixes
nx affected(#36569, #36568)@find(#36526)❤️ Thank You
v23.1.1Compare Source
23.1.1 (2026-07-30)
🚀 Features
🩹 Fixes
Configuration
📅 Schedule: Branch creation - "before 8am on monday" in timezone UTC, Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR has been generated by Renovate Bot.